Skip to content

docs: update custom error page - #13200

Open
halvaradop wants to merge 3 commits into
nextauthjs:mainfrom
halvaradop:docs/update-custom-error-page
Open

halvaradop wants to merge 3 commits into
nextauthjs:mainfrom
halvaradop:docs/update-custom-error-page

Conversation

@halvaradop

@halvaradop halvaradop commented Aug 28, 2025 •

Copy link
Copy Markdown
Contributor

☕️ Reasoning

The PR updates the Custom Error page by re-exporting the SignInPageErrorParam and ErrorPageParam types from the next-auth package.
This allows creating an object map of all possible errors that may occur during the authentication workflow.

🧢 Checklist

  • Documentation
  • Tests
  • Ready to be merged

🎫 Affected issues

📌 Resources

@halvaradop
halvaradop marked this pull request as draft August 28, 2025 23:54
@vercel

vercel Bot commented Aug 28, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Preview Comments Updated (UTC)
auth-docs Ready Ready Preview Comment Aug 29, 2025 8:51pm
1 Skipped Deployment
Project Deployment Preview Comments Updated (UTC)
next-auth-docs Ignored Ignored Preview Aug 29, 2025 8:51pm

@vercel

vercel Bot commented Aug 28, 2025

Copy link
Copy Markdown

@halvaradop is attempting to deploy a commit to the authjs Team on Vercel.

A member of the Team first needs to authorize it.

@codecov

codecov Bot commented Aug 29, 2025 •

Copy link
Copy Markdown

Codecov Report

❌ Patch coverage is 7.40741% with 25 lines in your changes missing coverage. Please review.
✅ Project coverage is 36.81%. Comparing base (39dd3b9) to head (9da84a3).

Files with missing lines Patch % Lines
packages/core/src/types.ts 0.00% 25 Missing ⚠️
Additional details and impacted files
@@            Coverage Diff             @@
##             main   #13200      +/-   ##
==========================================
- Coverage   38.76%   36.81%   -1.95%     
==========================================
  Files         200      197       -3     
  Lines       32323    31352     -971     
  Branches     1393     1249     -144     
==========================================
- Hits        12529    11543     -986     
- Misses      19794    19809      +15     

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

@halvaradop
halvaradop marked this pull request as ready for review August 29, 2025 20:42
@github-actions github-actions Bot added the core Refers to `@auth/core` label Aug 29, 2025

This branch was successfully deployed

1 active deployment
Preview – auth-docs — 9da84a3c Deployed Aug 29, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

core Refers to `@auth/core` next-auth

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant