Skip to content

frontend: Pod: Reorder unescaped string literals - #3371

Merged
illume merged 1 commit into
kubernetes-sigs:mainfrom
skoeva:unescaping
May 28, 2025
Merged

illume merged 1 commit into
kubernetes-sigs:mainfrom
skoeva:unescaping

Conversation

@skoeva

@skoeva skoeva commented May 27, 2025

Copy link
Copy Markdown
Contributor

This change reorders the unescaped string literals to ensure that any \ characters introduced by earlier replacements are not unintentionally unescaped again.

This change reorders the unescaped string literals to ensure that any \
characters introduced by earlier replacements are not unintentionally
unescaped again.
@skoeva skoeva self-assigned this May 27, 2025
@skoeva skoeva added the frontend Issues related to the frontend label May 27, 2025
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label May 27, 2025
@k8s-ci-robot
k8s-ci-robot requested review from vyncent-t and yolossn May 27, 2025 14:42
@k8s-ci-robot k8s-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label May 27, 2025
@illume illume added kind/bug Categorizes issue or PR as related to a bug. security labels May 28, 2025

@illume illume left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks! 🎉

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: illume, skoeva

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 28, 2025
@illume
illume merged commit 63d8b28 into kubernetes-sigs:main May 28, 2025
@skoeva
skoeva deleted the unescaping branch June 23, 2025 19:09
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. frontend Issues related to the frontend kind/bug Categorizes issue or PR as related to a bug. security size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

No open projects

Development

Successfully merging this pull request may close these issues.

4 participants