Skip to content

Bump API schema to de004dcd - #13931

Closed
getsantry[bot] wants to merge 1 commit into
masterfrom
bot/bump-api-schema-to-de004dcd
Closed

getsantry[bot] wants to merge 1 commit into
masterfrom
bot/bump-api-schema-to-de004dcd

Conversation

@getsantry

@getsantry getsantry Bot commented Jun 5, 2025

Copy link
Copy Markdown
Contributor

No description provided.

@getsantry
getsantry Bot enabled auto-merge (squash) June 5, 2025 14:34
@vercel

vercel Bot commented Jun 5, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

Name Status Preview Comments Updated (UTC)
develop-docs ✅ Ready (Inspect) Visit Preview 💬 Add feedback Jun 5, 2025 2:51pm
sentry-docs ✅ Ready (Inspect) Visit Preview 💬 Add feedback Jun 5, 2025 2:51pm
1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
changelog ⬜️ Ignored (Inspect) Jun 5, 2025 2:51pm

@codecov

codecov Bot commented Jun 5, 2025 •

Copy link
Copy Markdown

Bundle Report

Changes will decrease total bundle size by 15 bytes (-0.0%) ⬇️. This is within the configured threshold ✅

Detailed changes
Bundle name Size Change
sentry-docs-server-cjs 11.22MB -9 bytes (-0.0%) ⬇️
sentry-docs-client-array-push 9.77MB -6 bytes (-0.0%) ⬇️

Affected Assets, Files, and Routes:

view changes for bundle: sentry-docs-server-cjs

Assets Changed:

Asset Name Size Change Total Size Change (%)
1729.js -3 bytes 1.75MB -0.0%
../instrumentation.js -3 bytes 1.08MB -0.0%
9523.js -3 bytes 1.05MB -0.0%
view changes for bundle: sentry-docs-client-array-push

Assets Changed:

Asset Name Size Change Total Size Change (%)
static/chunks/pages/_app-*.js -3 bytes 868.7kB -0.0%
static/chunks/8165-*.js -3 bytes 410.1kB -0.0%
static/vTXwXSyvajv2bltcHKP5U/_buildManifest.js (New) 616 bytes 616 bytes 100.0% 🚀
static/vTXwXSyvajv2bltcHKP5U/_ssgManifest.js (New) 77 bytes 77 bytes 100.0% 🚀
static/GQiTQAg8_cTS36bNeMzBW/_buildManifest.js (Deleted) -616 bytes 0 bytes -100.0% 🗑️
static/GQiTQAg8_cTS36bNeMzBW/_ssgManifest.js (Deleted) -77 bytes 0 bytes -100.0% 🗑️

@AbhiPrasad AbhiPrasad closed this Jun 9, 2025
auto-merge was automatically disabled June 9, 2025 19:27

Pull request was closed

@github-actions github-actions Bot locked and limited conversation to collaborators Jun 25, 2025

This branch was successfully deployed

2 active deployments
Preview – sentry-docs — b69d2c67 Deployed Jun 5, 2025 by vercel[bot]
Preview – develop-docs — b69d2c67 Deployed Jun 5, 2025 by vercel[bot]
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant