Skip to content

CRM: Remove experimental automations module - #48422

Merged
tbradsha merged 11 commits into
trunkfrom
update/crm/remove_experimental_automations
May 1, 2026
Merged

tbradsha merged 11 commits into
trunkfrom
update/crm/remove_experimental_automations

Conversation

@tbradsha

@tbradsha tbradsha commented Apr 30, 2026

Copy link
Copy Markdown
Contributor

Proposed changes

This removes an experimental automations module that was never launched, along with its tests and config.

I left a few files related to the project in the src folder that are also not used (behind a feature flag) but have potential (WP REST support, the event manager, and object factories).

The package.json was down to one direct dependency (@wordpress/theme), and that just pulls in a CSS file that is imported by SCSS, so I moved that to be a dev dependency.

Related product discussion/links

Does this pull request change what data or activity we track or use?

Testing instructions

  1. CI should be happy. Things should build and tests should pass.
  2. Nothing removed should have been live. A quick smoke test is probably a good idea, though.
  3. Let's make sure the CSS import still works (see https://github.com/Automattic/jetpack/blob/update/crm/remove_experimental_automations/projects/plugins/crm/sass/emerald/_jpcrm_top_menu.scss#L1).

@tbradsha
tbradsha requested a review from a team April 30, 2026 19:36
@tbradsha tbradsha self-assigned this Apr 30, 2026
@tbradsha tbradsha added the [Status] Needs Review This PR is ready for review. label Apr 30, 2026
@github-actions

github-actions Bot commented Apr 30, 2026

Copy link
Copy Markdown
Contributor

Are you an Automattician? Please test your changes on all WordPress.com environments to help mitigate accidental explosions.

  • To test on WoA, go to the Plugins menu on a WoA dev site. Click on the "Upload" button and follow the upgrade flow to be able to upload, install, and activate the Jetpack Beta plugin. Once the plugin is active, go to Jetpack > Jetpack Beta, select your plugin (Jetpack or WordPress.com Site Helper), and enable the update/crm/remove_experimental_automations branch.
  • To test on Simple, run the following command on your sandbox:
bin/jetpack-downloader test jetpack update/crm/remove_experimental_automations
bin/jetpack-downloader test jetpack-mu-wpcom-plugin update/crm/remove_experimental_automations

Interested in more tips and information?

  • In your local development environment, use the jetpack rsync command to sync your changes to a WoA dev blog.
  • Read more about our development workflow here: PCYsg-eg0-p2
  • Figure out when your changes will be shipped to customers here: PCYsg-eg5-p2

@github-actions github-actions Bot added [CRM] Automations Module [Plugin] CRM Moved to https://github.com/Automattic/jetpack-crm [Tests] Includes Tests Admin Page React-powered dashboard under the Jetpack menu labels Apr 30, 2026
@github-actions

Copy link
Copy Markdown
Contributor

Thank you for your PR!

When contributing to Jetpack, we have a few suggestions that can help us test and review your patch:

  • ✅ Include a description of your PR changes.
  • ✅ Add a "[Status]" label (In Progress, Needs Review, ...).
  • ✅ Add testing instructions.
  • ✅ Specify whether this PR includes any changes to data or privacy.
  • ✅ Add changelog entries to affected projects

This comment will be updated as you work on your PR and make changes. If you think that some of those checks are not needed for your PR, please explain why you think so. Thanks for cooperation 🤖


Follow this PR Review Process:

  1. Ensure all required checks appearing at the bottom of this PR are passing.
  2. Make sure to test your changes on all platforms that it applies to. You're responsible for the quality of the code you ship.
  3. You can use GitHub's Reviewers functionality to request a review.
  4. When it's reviewed and merged, you will be pinged in Slack to deploy the changes to WordPress.com simple once the build is done.

If you have questions about anything, reach out in #jetpack-developers for guidance!

@jp-launch-control

jp-launch-control Bot commented Apr 30, 2026

Copy link
Copy Markdown

Code Coverage Summary

Coverage changed in 6 files. Only the first 5 are listed here.

File Coverage Δ% Δ Uncovered
projects/plugins/crm/src/rest-api/v4/class-rest-base-controller.php 1/11 (9.09%) -63.64% 7 💔
projects/plugins/crm/includes/ZeroBSCRM.AdminPages.php 0/1090 (0.00%) 0.00% -1 💚
projects/plugins/crm/includes/ZeroBSCRM.Core.php 22/946 (2.33%) 0.00% -2 💚
projects/plugins/crm/includes/ZeroBSCRM.REST.php 0/142 (0.00%) 0.00% -2 💚
projects/plugins/crm/includes/ZeroBSCRM.Database.php 0/352 (0.00%) 0.00% -6 💚

Full summary · PHP report · JS report

If appropriate, add one of these labels to override the failing coverage check: Covered by non-unit tests Use to ignore the Code coverage requirement check when E2Es or other non-unit tests cover the code Coverage tests to be added later Use to ignore the Code coverage requirement check when tests will be added in a follow-up PR I don't care about code coverage for this PR Use this label to ignore the check for insufficient code coveage.

@tbradsha
tbradsha force-pushed the update/crm/remove_experimental_automations branch from 6542e66 to 9a5ee40 Compare April 30, 2026 20:42

@anomiex anomiex left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I see a few more deps that seem like they can go away.

As for the PR itself... Seems like more of a product decision. 🤷

Comment thread projects/plugins/crm/package.json Outdated
Comment on lines 24 to 26
@@ -47,12 +25,7 @@
"@babel/preset-env": "7.29.2",
"@babel/runtime": "7.29.2",

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks like nothing needs these babel packages anymore.

Suggested change
"@babel/runtime": "7.29.2",

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Comment thread projects/plugins/crm/package.json Outdated
"@wordpress/theme": "0.11.0",
"chart.js": "4.5.0",
"copy-webpack-plugin": "14.0.0",
"css-loader": "6.11.0",

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

css-loader too.

Suggested change
"css-loader": "6.11.0",

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@tbradsha
tbradsha merged commit c42f006 into trunk May 1, 2026
95 of 96 checks passed
@tbradsha
tbradsha deleted the update/crm/remove_experimental_automations branch May 1, 2026 20:20
@github-actions github-actions Bot removed the [Status] Needs Review This PR is ready for review. label May 1, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Admin Page React-powered dashboard under the Jetpack menu [CRM] Automations Module [Plugin] CRM Moved to https://github.com/Automattic/jetpack-crm [Tests] Includes Tests

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants